Skip to content

ActionList: Ensure variant is respected when full - #4706

Merged
TylerJDev merged 5 commits into
mainfrom
tylerjdev/fix-for-variant-action-list
Jul 3, 2024
Merged

ActionList: Ensure variant is respected when full#4706
TylerJDev merged 5 commits into
mainfrom
tylerjdev/fix-for-variant-action-list

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Jun 28, 2024

Copy link
Copy Markdown
Member

When ActionList utilized the full variant, the style was not applying correctly (behind feature flag). This fix ensures that the variant is respected when the FF is enabled.

This change will only affect ActionList behind its feature flag (primer_react_action_list_item_as_button).

Changelog

Changed

  • Adds conditional to ensure variant is respected when full and FF is enabled

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Jun 28, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1f96ef5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Jun 28, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js91.11 KB (+0.16% 🔺)
packages/react/dist/browser.umd.js91.29 KB (+0.06% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4706 June 28, 2024 23:07 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/331186

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4706 June 29, 2024 00:46 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4706 July 1, 2024 13:53 Inactive
@TylerJDev
TylerJDev marked this pull request as ready for review July 1, 2024 14:08
@TylerJDev
TylerJDev requested a review from a team as a code ownerJuly 1, 2024 14:08
@TylerJDev
TylerJDev requested a review from joshblackJuly 1, 2024 14:08

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Just wanted to ask if there was a good way to add a test for this kind of change. Is this something that VRT could help out with or would this be more of a Jest thing?

@TylerJDev

Copy link
Copy Markdown
MemberAuthor

Just wanted to ask if there was a good way to add a test for this kind of change. Is this something that VRT could help out with or would this be more of a Jest thing?

That's a good point! I think a VRT would be perfect! I believe this wasn't caught because we don't test for variant: full in our VRT e2e tests. I'll follow up with a story that utilizes variant: full.

@TylerJDev
TylerJDev requested a review from a team as a code ownerJuly 1, 2024 16:22
@TylerJDev
TylerJDev requested a review from tbenningJuly 1, 2024 16:22
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4706 July 1, 2024 16:26 Inactive
@TylerJDev
TylerJDev added this pull request to the merge queueJul 3, 2024
Merged via the queue into main with commit 71859edJul 3, 2024
@TylerJDev
TylerJDev deleted the tylerjdev/fix-for-variant-action-list branch July 3, 2024 17:30
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@TylerJDev@mperrotti@joshblack