Skip to content

feat(react): add ScrollableRegion and useOverflow - #4719

Merged
joshblack merged 5 commits into
mainfrom
feat/export-scrollable-reigon-use-overflow
Jul 25, 2024
Merged

feat(react): add ScrollableRegion and useOverflow#4719
joshblack merged 5 commits into
mainfrom
feat/export-scrollable-reigon-use-overflow

Conversation

@joshblack

@joshblackjoshblack commented Jul 3, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3452

This adds the ScrollableRegion and useOverflow helpers to the experimental entry point. These helpers are used internally by our Table and Dialog components. It would be great to try them out to see if they're helpful more broadly or if they should stay on as experimental components.

Changelog

New

  • Add experimental ScrollableRegion component
  • Add experimental useOverflow hook

Changed

Removed

Rollout strategy

  • Minor release

@changeset-bot

changeset-botBot commented Jul 3, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d1270a6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4719 July 3, 2024 18:08 Inactive
@github-actions

github-actionsBot commented Jul 3, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js92.44 KB (+0.09% 🔺)
packages/react/dist/browser.umd.js92.73 KB (+0.05% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4719 July 3, 2024 20:56 Inactive
@joshblack
joshblack marked this pull request as ready for review July 23, 2024 15:31
@joshblack
joshblack requested a review from a team as a code ownerJuly 23, 2024 15:31
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4719 July 23, 2024 15:34 Inactive
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@broccolinisoup