Skip to content

Promote PageHeader to Beta 🎉 - #4725

Merged
broccolinisoup merged 8 commits into
mainfrom
promote-pageheader-beta
Jul 19, 2024
Merged

Promote PageHeader to Beta 🎉#4725
broccolinisoup merged 8 commits into
mainfrom
promote-pageheader-beta

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Jul 9, 2024

Copy link
Copy Markdown
Member

We completed the PageHeader work and closed out the epic (internal link only)

Changelog

Changed

  • Move PageHeader to Beta status and update the component accordingly.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

@changeset-bot

changeset-botBot commented Jul 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b89c7f5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Jul 9, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js92.36 KB (+1.26% 🔺)
packages/react/dist/browser.umd.js92.66 KB (+1.22% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4725 July 9, 2024 02:31 Inactive
@broccolinisoupbroccolinisoup changed the title Promote PageHeader to BetaPromote PageHeader to Beta 🎉Jul 9, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4725 July 9, 2024 02:47 Inactive
@broccolinisoupbroccolinisoup added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 10, 2024
@broccolinisoupbroccolinisoup removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 15, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4725 July 15, 2024 03:27 Inactive
@broccolinisoup
broccolinisoup marked this pull request as ready for review July 15, 2024 03:37
@broccolinisoup
broccolinisoup requested a review from a team as a code ownerJuly 15, 2024 03:37
Comment threadpackages/react/src/PageHeader/PageHeader.docs.json
@broccolinisoupbroccolinisoup changed the title Promote PageHeader to Beta 🎉Promote PageHeader to Alpha 🎉Jul 15, 2024
Comment threadpackages/react/src/PageHeader/PageHeader.docs.json Outdated
Comment thread.changeset/nice-forks-joke.md
@broccolinisoupbroccolinisoup added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 15, 2024
@broccolinisoupbroccolinisoup removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 15, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4725 July 15, 2024 05:29 Inactive

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🎉

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!

Could even promote to beta

Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@broccolinisoup@siddharthkp@joshblack