Skip to content

experimental/SelectPanel v2: Fix heading color - #4757

Merged
siddharthkp merged 3 commits into
mainfrom
selectpanel-v2-title-color
Jul 22, 2024
Merged

experimental/SelectPanel v2: Fix heading color#4757
siddharthkp merged 3 commits into
mainfrom
selectpanel-v2-title-color

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Jul 19, 2024

Copy link
Copy Markdown
Member

Changelog

<dialog> ships with a default browser style (color: canvastext) that need to be overwritten to work across our themes

BeforeAfter
before, text barely visibleafter, text now clearly visible

Rollout strategy

  • Patch release

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Jul 19, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b3ceb77

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Jul 19, 2024
@siddharthkpsiddharthkp self-assigned this Jul 19, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review July 19, 2024 13:10
@siddharthkp
siddharthkp requested a review from a team as a code ownerJuly 19, 2024 13:10
@github-actions

github-actionsBot commented Jul 19, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js92.42 KB (0%)
packages/react/dist/browser.umd.js92.68 KB (0%)

@siddharthkpsiddharthkp added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 22, 2024
@primer
primerBot requested a review from a team as a code ownerJuly 22, 2024 08:34
@primer
primerBot requested a review from mperrottiJuly 22, 2024 08:34
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 22, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4757 July 22, 2024 08:38 Inactive
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@siddharthkp@joshblack@langermank