Skip to content

StateLabel: Add labels to icon - #4764

Merged
siddharthkp merged 12 commits into
mainfrom
labelled-state-labels
Jul 24, 2024
Merged

StateLabel: Add labels to icon#4764
siddharthkp merged 12 commits into
mainfrom
labelled-state-labels

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Jul 22, 2024

Copy link
Copy Markdown
Member

The issueClosedNotPlanned variant has an icon that differentiates it from issueClosed and pullClosed . The icon has no text alternative and includes aria-hidden="true", so is ignored by screen readers. Screen reader users do not know what the labels 'Draft', 'Open' and 'Closed' refers to, unless context is being provided elsewhere on the page, for example, the component is on an Issue page.
Screen shot of State Label variants for Issues and Pull requests, each consists of a single word, with a coloured background shaped like a pill, with an icon

Before:

<spanclass="StateLabel__StateLabelBase-sc-qthdln-0 eDFXzm"><svgaria-hidden="true" focusable="false">...</svg>
Closed
</span>

After:

<spanclass="StateLabel__StateLabelBase-sc-qthdln-0 eDFXzm"><svgrole="img" aria-label="Issue" focusable="false">...</svg>
Closed
</span>

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Jul 22, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e7a3e29

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Jul 22, 2024
@siddharthkpsiddharthkp self-assigned this Jul 22, 2024
@github-actionsgithub-actionsBot added the staff Author is a staff member label Jul 22, 2024
@siddharthkpsiddharthkp added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 22, 2024
@siddharthkpsiddharthkp mentioned this pull request Jul 22, 2024
@github-actions

github-actionsBot commented Jul 22, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js92.47 KB (+0.12% 🔺)
packages/react/dist/browser.umd.js92.68 KB (-0.01% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4764 July 22, 2024 14:25 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 22, 2024
@primer-integration

primer-integrationBot commented Jul 22, 2024

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/333917

Update from Sid: Integration tests are passing :)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4764 July 22, 2024 16:23 Inactive
@siddharthkpsiddharthkp added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Jul 23, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review July 23, 2024 10:05
@siddharthkp
siddharthkp requested a review from a team as a code ownerJuly 23, 2024 10:05

@TylerJDevTylerJDev left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! ✨

issueOpened: 'Issue',
pullOpened: 'Pull request',
issueClosed: 'Issue',
issueClosedNotPlanned: 'Issue',

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

non-blocking, but would we want to add "not planned" here? (i.e. "Issue, not planned")

@siddharthkpsiddharthkpJul 24, 2024

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Great catch! Thank you ❤️

@siddharthkp
siddharthkp added this pull request to the merge queueJul 24, 2024
@siddharthkp
siddharthkp removed this pull request from the merge queue due to a manual request Jul 24, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

accessibilityintegration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@TylerJDev