Skip to content

Bug fix: ActionList divider disappears on selected items - #4778

Merged
langermank merged 4 commits into
mainfrom
fix-actionlist-hover-divider
Jul 25, 2024
Merged

Bug fix: ActionList divider disappears on selected items#4778
langermank merged 4 commits into
mainfrom
fix-actionlist-hover-divider

Conversation

@langermank

Copy link
Copy Markdown
Contributor

@changeset-bot

changeset-botBot commented Jul 25, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3c57fa0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 25, 2024
@github-actions

github-actionsBot commented Jul 25, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js94.95 KB (+0.07% 🔺)
packages/react/dist/browser.umd.js95.2 KB (-0.02% 🔽)

@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 25, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4778 July 25, 2024 21:21 Inactive
@langermank
langermank marked this pull request as ready for review July 25, 2024 21:33
@langermank
langermank requested a review from a team as a code ownerJuly 25, 2024 21:33
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@joshblack