Skip to content

refactor: move drafts to experimental - #4785

Merged
joshblack merged 15 commits into
mainfrom
refactor/move-drafts-to-experimental-2
Sep 16, 2024
Merged

refactor: move drafts to experimental#4785
joshblack merged 15 commits into
mainfrom
refactor/move-drafts-to-experimental-2

Conversation

@joshblack

@joshblackjoshblack commented Jul 26, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3666

Now that https://github.com/github/primer/issues/3668 has been addressed, we can officially remove drafts from @primer/react 🫡

Changelog

New

Changed

Removed

  • Remove drafts entrypoint in favor of experimental

Rollout strategy

  • Major release; if selected, include a written rollout or migration plan

Migration plan

Testing & Reviewing

  • Verify the drafts entrypoint has been removed
  • Verify that no mentions of drafts exists in documentation
  • Verify that experimental continues to match draft entrypoint now that drafts is gone

@changeset-bot

changeset-botBot commented Jul 26, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 086ac50

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Jul 26, 2024
@github-actions

github-actionsBot commented Jul 26, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.38 KB (-0.06% 🔽)
packages/react/dist/browser.umd.js97.64 KB (-0.09% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4785 September 12, 2024 16:47 Inactive
@joshblack
joshblack marked this pull request as ready for review September 12, 2024 16:48
@joshblack
joshblack requested a review from a team as a code ownerSeptember 12, 2024 16:48

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Trust you with this!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4785 September 16, 2024 15:05 Inactive
@joshblack
joshblack added this pull request to the merge queueSep 16, 2024
Merged via the queue into main with commit b74c47fSep 16, 2024
@joshblack
joshblack deleted the refactor/move-drafts-to-experimental-2 branch September 16, 2024 15:32
@primerprimerBot mentioned this pull request Sep 16, 2024
TylerJDev pushed a commit that referenced this pull request Sep 23, 2024
* refactor: move drafts to experimental
* chore: add changeset
* refactor: rename drafts to experimental
* chore: remove imports to drafts
* test(e2e): move drafts to experimental
* docs: change drafts to experimental
* chore: update changeset
* chore: update story ids
* docs: update drafts to experimental
* chore: update top-level storybook order
* chore: update banner e2e tests
* chore: update figma file imports
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@siddharthkp