Skip to content

refactor(react): move deprecated components to deprecated - #4798

Merged
joshblack merged 2 commits into
mainfrom
refactor/move-tooltip-v1-to-deprecated
Jul 31, 2024
Merged

refactor(react): move deprecated components to deprecated#4798
joshblack merged 2 commits into
mainfrom
refactor/move-tooltip-v1-to-deprecated

Conversation

@joshblack

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3679

Changelog

New

Changed

  • Move deprecated components to deprecated entrypoint

Removed

Rollout strategy

  • Minor release

@changeset-bot

changeset-botBot commented Jul 30, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: aaebc6a

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Jul 30, 2024
@joshblack
joshblack marked this pull request as ready for review July 30, 2024 18:12
@joshblack
joshblack requested a review from a team as a code ownerJuly 30, 2024 18:12
@joshblack
joshblack requested a review from TylerJDevJuly 30, 2024 18:12
@github-actions

github-actionsBot commented Jul 30, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js95.76 KB (0%)
packages/react/dist/browser.umd.js96.06 KB (0%)

@joshblack
joshblack added this pull request to the merge queueJul 31, 2024
Merged via the queue into main with commit 0fa60a4Jul 31, 2024
@joshblack
joshblack deleted the refactor/move-tooltip-v1-to-deprecated branch July 31, 2024 16:23
@primerprimerBot mentioned this pull request Jul 31, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@TylerJDev