Skip to content

feat(react): enable support for CSS Modules and emitting *.css files in bundle - #4800

Merged
joshblack merged 10 commits into
mainfrom
feat/add-css-modules-to-primer-react
Aug 1, 2024
Merged

feat(react): enable support for CSS Modules and emitting *.css files in bundle#4800
joshblack merged 10 commits into
mainfrom
feat/add-css-modules-to-primer-react

Conversation

@joshblack

@joshblackjoshblack commented Jul 31, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3685

Changelog

New

  • Add support for emitting *.css files alongside *.js files in order to allow authoring styles in CSS Modules

Changed

Removed

Rollout strategy

  • Major release; if selected, include a written rollout or migration plan

We've been testing out support for this incrementally in dotcom. As a result, we've landed some upstream PRs to enable support for this work. In addition, we will try out a feature flag approach to test out style-only changes before rolling them out more broadly.

@changeset-bot

changeset-botBot commented Jul 31, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e877f6d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Jul 31, 2024
@github-actions

github-actionsBot commented Jul 31, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js95.84 KB (0%)
packages/react/dist/browser.umd.js96.18 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4800 July 31, 2024 15:17 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4800 July 31, 2024 15:29 Inactive
@joshblack
joshblack marked this pull request as ready for review July 31, 2024 15:53
@joshblack
joshblack requested a review from a team as a code ownerJuly 31, 2024 15:53
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4800 July 31, 2024 16:00 Inactive

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

YAY

@joshblack
joshblack enabled auto-merge July 31, 2024 20:03
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4800 July 31, 2024 20:04 Inactive
@joshblack
joshblack added this pull request to the merge queueAug 1, 2024
Merged via the queue into main with commit 482b4d6Aug 1, 2024
@joshblack
joshblack deleted the feat/add-css-modules-to-primer-react branch August 1, 2024 18:44
@primerprimerBot mentioned this pull request Aug 1, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@langermank