Skip to content

Add variant prop to Heading - #4806

Merged
joshblack merged 23 commits into
mainfrom
heading-component-style-props
Aug 7, 2024
Merged

Add variant prop to Heading#4806
joshblack merged 23 commits into
mainfrom
heading-component-style-props

Conversation

@langermank

@langermanklangermank commented Aug 1, 2024

Copy link
Copy Markdown
Contributor

UI mob pair session 💅

Changelog

New

  • Adds new variant prop for styling small, medium, and large headings

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@langermank
langermank requested a review from a team as a code ownerAugust 1, 2024 17:30
@changeset-bot

changeset-botBot commented Aug 1, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e4ad72f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Aug 1, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.05 KB (+0.27% 🔺)
packages/react/dist/browser.umd.js96.39 KB (+0.21% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4806 August 1, 2024 17:35 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 1, 2024
@primer
primerBot requested a review from a team as a code ownerAugust 1, 2024 18:26
@primer
primerBot requested a review from maximedegreveAugust 1, 2024 18:26
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 1, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4806 August 1, 2024 18:31 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4806 August 1, 2024 21:42 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 2, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 2, 2024
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 5, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4806 August 5, 2024 16:13 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 5, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4806 August 5, 2024 16:22 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 6, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4806 August 6, 2024 15:59 Inactive
@langermanklangermank added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Aug 6, 2024
@@ -0,0 +1,5 @@
---
"@primer/react": patch

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Suggested change
"@primer/react": patch
"@primer/react": minor

@joshblack
joshblack added this pull request to the merge queueAug 7, 2024
Merged via the queue into main with commit 84d1604Aug 7, 2024
@joshblack
joshblack deleted the heading-component-style-props branch August 7, 2024 18:05
@primerprimerBot mentioned this pull request Aug 7, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

UI mob pairupdate snapshots🤖 Command that updates VRT snapshots on the pull request

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@langermank@siddharthkp@mperrotti@joshblack