Skip to content

Update TextInput.TrailingAction hover state - #4831

Merged
siddharthkp merged 5 commits into
mainfrom
textinput-action-hover-design
Aug 20, 2024
Merged

Update TextInput.TrailingAction hover state#4831
siddharthkp merged 5 commits into
mainfrom
textinput-action-hover-design

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Aug 8, 2024

Copy link
Copy Markdown
Member

Changelog

  1. Styles: Remove icon button background, change icon color on hover
  2. Stories: In storybook example for "Clear input", hide button when the text input is already empty

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Aug 8, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c59922d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Aug 8, 2024
@siddharthkpsiddharthkp changed the title Textinput action hover designFix TextInput action hover stateAug 8, 2024
@siddharthkpsiddharthkp self-assigned this Aug 8, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review August 8, 2024 13:41
@siddharthkp
siddharthkp requested a review from a team as a code ownerAugust 8, 2024 13:41
@siddharthkpsiddharthkp changed the title Fix TextInput action hover stateUpdate TextInput.TrailingAction hover stateAug 8, 2024
@github-actions

github-actionsBot commented Aug 8, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.01 KB (+0.02% 🔺)
packages/react/dist/browser.umd.js96.34 KB (+0.04% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4831 August 8, 2024 13:44 Inactive
@siddharthkp
siddharthkp requested review from mperrotti and removed request for camertronAugust 9, 2024 09:45
@siddharthkpsiddharthkp added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 9, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4831 August 9, 2024 10:13 Inactive
@primer
primerBot requested a review from a team as a code ownerAugust 9, 2024 10:17
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 9, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4831 August 9, 2024 10:22 Inactive
@siddharthkp
siddharthkp requested review from langermank and removed request for maximedegreve and mperrottiAugust 16, 2024 08:41
@siddharthkp
siddharthkp requested review from maximedegreve and removed request for langermankAugust 16, 2024 08:42
@siddharthkp
siddharthkp added this pull request to the merge queueAug 20, 2024
Merged via the queue into main with commit 0d7a02aAug 20, 2024
@siddharthkp
siddharthkp deleted the textinput-action-hover-design branch August 20, 2024 10:32
@primerprimerBot mentioned this pull request Aug 20, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

TextInput action hover state is incorrect

3 participants

@siddharthkp@maximedegreve@joshblack