Skip to content

feat(Banner): add support for custom icons when variant="upsell" - #4838

Merged
joshblack merged 4 commits into
mainfrom
joshblack-patch-1
Aug 13, 2024
Merged

feat(Banner): add support for custom icons when variant="upsell"#4838
joshblack merged 4 commits into
mainfrom
joshblack-patch-1

Conversation

@joshblack

@joshblackjoshblack commented Aug 12, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3784

Update Banner to support customizing icons with both info and upsell

Changelog

New

  • Add support for custom icons when Banner variant="upsell"

Changed

Removed

Rollout strategy

  • Minor release

@changeset-bot

changeset-botBot commented Aug 12, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ede98f8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Aug 12, 2024
@github-actions

github-actionsBot commented Aug 12, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.2 KB (0%)
packages/react/dist/browser.umd.js96.46 KB (0%)

@joshblackjoshblack changed the title Update Banner.tsxfeat(Banner): add support for custom icons when variant="upsell"Aug 12, 2024
@joshblack
joshblack marked this pull request as ready for review August 12, 2024 16:48
@joshblack
joshblack requested a review from a team as a code ownerAugust 12, 2024 16:48
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4838 August 12, 2024 16:50 Inactive

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!

@joshblack
joshblack added this pull request to the merge queueAug 13, 2024
Merged via the queue into main with commit 7a3b55dAug 13, 2024
@joshblack
joshblack deleted the joshblack-patch-1 branch August 13, 2024 15:19
@primerprimerBot mentioned this pull request Aug 12, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@siddharthkp