Skip to content

ProgressBar: Remove aria-busy from ProgressBar - #4841

Merged
TylerJDev merged 2 commits into
mainfrom
tylerjdev/remove-aria-busy
Aug 14, 2024
Merged

ProgressBar: Remove aria-busy from ProgressBar#4841
TylerJDev merged 2 commits into
mainfrom
tylerjdev/remove-aria-busy

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Aug 12, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3360

It was noted that the usage of aria-busy on role="progressbar" is unnecessary, as it is intended for a section of the page that is currently being updated, signaling to screen readers that they should wait before announcing the content.

This PR removes the attribute from ProgressBar.

Changelog

New

Removed

  • Removed aria-busy from ProgressBar

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Aug 12, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3bf029a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Aug 12, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.04 KB (-0.17% 🔽)
packages/react/dist/browser.umd.js96.44 KB (-0.02% 🔽)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/337288

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4841 August 13, 2024 13:25 Inactive
@TylerJDev
TylerJDev marked this pull request as ready for review August 13, 2024 13:46
@TylerJDev
TylerJDev requested a review from a team as a code ownerAugust 13, 2024 13:46
@TylerJDevTylerJDev changed the title Remove aria-busy from ProgressBarProgressBar: Remove aria-busy from ProgressBarAug 13, 2024
@TylerJDev
TylerJDev added this pull request to the merge queueAug 14, 2024
Merged via the queue into main with commit f3b08dfAug 14, 2024
@TylerJDev
TylerJDev deleted the tylerjdev/remove-aria-busy branch August 14, 2024 19:15
@primerprimerBot mentioned this pull request Aug 14, 2024
jonrohan pushed a commit that referenced this pull request Aug 15, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@joshblack