Skip to content

Update Label font weight to match Rails component - #4865

Merged
camertron merged 4 commits into
mainfrom
update_label_font_weight
Aug 23, 2024
Merged

Update Label font weight to match Rails component#4865
camertron merged 4 commits into
mainfrom
update_label_font_weight

Conversation

@camertron

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/3871

BeforeAfter
A screenshot of the existing Label component with a font weight of 600A screenshot of the modified Label component with a font weight of 500

Changelog

No API changes.

New

Changed

The font weight of Label components has been changed from bold to semibold to match the Rails component.

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

None, this change is visual-only.

Merge checklist

- [ ] Added/updated tests
- [ ] Added/updated documentation
- [ ] Added/updated previews (Storybook)
- [ ] Changes are SSR compatible
- [ ] Tested in Chrome
- [ ] Tested in Firefox
- [ ] Tested in Safari
- [ ] Tested in Edge
- [ ] (GitHub staff only) Integration tests pass at github/github (Learn more about how to run integration tests)

@changeset-bot

changeset-botBot commented Aug 20, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c011dd2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@camertroncamertron changed the title Update Label font weight to match Rails componentUpdate Label font weight to match Rails componentAug 20, 2024
@github-actions

github-actionsBot commented Aug 20, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.29 KB (+0.1% 🔺)
packages/react/dist/browser.umd.js96.57 KB (+0.03% 🔺)

@camertroncamertron added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 21, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 21, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4865 August 21, 2024 16:52 Inactive
@camertron
camertron marked this pull request as ready for review August 21, 2024 17:01
@camertron
camertron requested review from a team as code ownersAugust 21, 2024 17:01
@camertron
camertron enabled auto-merge August 22, 2024 18:47
@camertron
camertron disabled auto-merge August 22, 2024 18:47
@camertron
camertron enabled auto-merge August 22, 2024 20:14
@camertron
camertron added this pull request to the merge queueAug 23, 2024
Merged via the queue into main with commit ce2c674Aug 23, 2024
@camertron
camertron deleted the update_label_font_weight branch August 23, 2024 18:18
@primerprimerBot mentioned this pull request Aug 23, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@camertron@jonrohan@mperrotti@anandfresh