Skip to content

Add className to Blankslate - #4866

Merged
langermank merged 3 commits into
mainfrom
add-classname-blankslate
Aug 21, 2024
Merged

Add className to Blankslate#4866
langermank merged 3 commits into
mainfrom
add-classname-blankslate

Conversation

@langermank

@langermanklangermank commented Aug 20, 2024

Copy link
Copy Markdown
Contributor

Add className prop to support CSS Modules

@changeset-bot

changeset-botBot commented Aug 20, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: bb63e3a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Aug 20, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.2 KB (0%)
packages/react/dist/browser.umd.js96.54 KB (0%)

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

👍🏻

@langermank
langermank added this pull request to the merge queueAug 21, 2024
Merged via the queue into main with commit 587603bAug 21, 2024
@langermank
langermank deleted the add-classname-blankslate branch August 21, 2024 00:34
@primerprimerBot mentioned this pull request Aug 21, 2024
@dylanatsmith

Copy link
Copy Markdown
Member

I needed this so bad today 😂❤️

@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@langermank@dylanatsmith@jonrohan