Skip to content

StackItem responsive grow bug fix - #4891

Merged
langermank merged 5 commits into
mainfrom
stack-bug-fix
Aug 28, 2024
Merged

StackItem responsive grow bug fix#4891
langermank merged 5 commits into
mainfrom
stack-bug-fix

Conversation

@langermank

@langermanklangermank commented Aug 27, 2024

Copy link
Copy Markdown
Contributor

StackItem wasn't respecting the grow property for responsive values due to the boolean nature. This PR leaves the boolean in place, but returns a true or false string in the data-attribute instead.

@changeset-bot

changeset-botBot commented Aug 27, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2cf445e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Aug 27, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.34 KB (+0.01% 🔺)
packages/react/dist/browser.umd.js96.66 KB (+0.05% 🔺)

Comment threadpackages/react/src/Stack/__tests__/StackItem.test.tsx Outdated
@langermank
langermank added this pull request to the merge queueAug 28, 2024
Merged via the queue into main with commit ae00350Aug 28, 2024
@langermank
langermank deleted the stack-bug-fix branch August 28, 2024 18:26
@primerprimerBot mentioned this pull request Aug 28, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@mperrotti