Skip to content

Bump primer/primitives v9.0.3 - #4895

Merged
langermank merged 2 commits into
mainfrom
bump-prim
Sep 4, 2024
Merged

Bump primer/primitives v9.0.3#4895
langermank merged 2 commits into
mainfrom
bump-prim

Conversation

@langermank

Copy link
Copy Markdown
Contributor

Fixing a bug with the danger button.

@changeset-bot

changeset-botBot commented Aug 29, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3c36db7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 29, 2024
@github-actions

github-actionsBot commented Aug 29, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.34 KB (0%)
packages/react/dist/browser.umd.js96.66 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4895 August 29, 2024 20:50 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 29, 2024
@langermank
langermank marked this pull request as ready for review August 29, 2024 22:25
@langermank
langermank requested a review from a team as a code ownerAugust 29, 2024 22:25
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 29, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 29, 2024
@langermank
langermank added this pull request to the merge queueSep 4, 2024
Merged via the queue into main with commit 0208e9eSep 4, 2024
@langermank
langermank deleted the bump-prim branch September 4, 2024 16:28
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@langermank@joshblack@anandfresh