Skip to content

(waiting) Select Panel: Add home and end keys for keyboard navigation - #4903

Closed
siddharthkp wants to merge 68 commits into
mainfrom
select-panel--keyboard-add-home-and-end
Closed

(waiting) Select Panel: Add home and end keys for keyboard navigation#4903
siddharthkp wants to merge 68 commits into
mainfrom
select-panel--keyboard-add-home-and-end

Conversation

@siddharthkp

Copy link
Copy Markdown
Member

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e33bc33

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.93 KB (+0.58% 🔺)
packages/react/dist/browser.umd.js97.12 KB (+0.42% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4903 September 3, 2024 14:32 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/340195

@siddharthkpsiddharthkp changed the title Select Panel: Add home and end keys for keyboard navigation(waiting) Select Panel: Add home and end keys for keyboard navigationSep 3, 2024
@siddharthkp

Copy link
Copy Markdown
MemberAuthor

Copied over to #4906

@siddharthkp
siddharthkp deleted the select-panel--keyboard-add-home-and-end branch October 4, 2024 13:42
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@broccolinisoup