Skip to content

LabelGroup: Update "toggle" accessible name to include visual text - #4910

Merged
TylerJDev merged 2 commits into
mainfrom
tylerjdev/label-group-accName
Sep 13, 2024
Merged

LabelGroup: Update "toggle" accessible name to include visual text#4910
TylerJDev merged 2 commits into
mainfrom
tylerjdev/label-group-accName

Conversation

@TylerJDev

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3426

Changelog

Changed

  • Updates the accessible name of the "toggle" button in LabelGroup to include the visual label

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@TylerJDevTylerJDev self-assigned this Sep 3, 2024
@TylerJDev
TylerJDev requested a review from a team as a code ownerSeptember 3, 2024 19:34
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: bdc650c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.42 KB (+0.06% 🔺)
packages/react/dist/browser.umd.js96.66 KB (-0.06% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4910 September 3, 2024 19:38 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/340287

) : hiddenItemIds.length ? (
<Button ref={expandButtonRef} variant="invisible" size="small" onClick={showAllTokensInline}>
<VisuallyHidden>Show all {totalLength}</VisuallyHidden>
<VisuallyHidden>Show +{hiddenItemIds.length} more</VisuallyHidden>

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Random question, what does including + here do? Is this changing how it's announced?

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

It doesn't really change much. For all the screen readers I tested with (NVDA, VoiceOver, JAWS) they all announced "Show plus 11 more". It would be preferable for it to say "11 plus", but then we'd need to move the + in the visual text to be after the number (e.g. 11+).

@TylerJDev
TylerJDev added this pull request to the merge queueSep 13, 2024
Merged via the queue into main with commit c2e4d5eSep 13, 2024
@TylerJDev
TylerJDev deleted the tylerjdev/label-group-accName branch September 13, 2024 16:10
@primerprimerBot mentioned this pull request Sep 13, 2024
TylerJDev added a commit that referenced this pull request Sep 23, 2024
…4910)
* Update `LabelGroup` truncate button accName
* Add changeset
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@joshblack