Skip to content

SegmentedControl: Add context to ariaLabel prop - #4911

Merged
TylerJDev merged 3 commits into
mainfrom
tylerjdev/segmented-control-accName
Sep 4, 2024
Merged

SegmentedControl: Add context to ariaLabel prop#4911
TylerJDev merged 3 commits into
mainfrom
tylerjdev/segmented-control-accName

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Sep 3, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3497

Changelog

Changed

  • Updates accessible name of SegmentedControl menu button to contain the visual text, and provided aria-label.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Sep 3, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 137634c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Sep 3, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.41 KB (+0.16% 🔺)
packages/react/dist/browser.umd.js96.65 KB (+0.03% 🔺)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/340353

@TylerJDev
TylerJDev marked this pull request as ready for review September 4, 2024 02:58
@TylerJDev
TylerJDev requested a review from a team as a code ownerSeptember 4, 2024 02:58

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! Wanted to ask if it would be possible to include a test too to make sure we don't miss this in the future? 😅

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4911 September 4, 2024 19:41 Inactive
@TylerJDev
TylerJDev added this pull request to the merge queueSep 4, 2024
Merged via the queue into main with commit 9846375Sep 4, 2024
@TylerJDev
TylerJDev deleted the tylerjdev/segmented-control-accName branch September 4, 2024 20:12
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@joshblack