Skip to content

Moving Link CSS modules to staff feature flag - #4916

Merged
jonrohan merged 2 commits into
mainfrom
css_modules_link_staff
Sep 4, 2024
Merged

Moving Link CSS modules to staff feature flag#4916
jonrohan merged 2 commits into
mainfrom
css_modules_link_staff

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3719

Link has been team shipped for a week without any issues, promoting to staff ship based on lifecycle. https://github.com/github/primer-engineering/discussions/138

Changelog

Changed

Link CSS modules behind primer_react_css_modules_staff feature flag.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerSeptember 3, 2024 23:34
@changeset-bot

changeset-botBot commented Sep 3, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a9db5d3

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 3, 2024
@github-actions

github-actionsBot commented Sep 3, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.38 KB (+0.01% 🔺)
packages/react/dist/browser.umd.js96.63 KB (-0.1% 🔽)

@jonrohan
jonrohan added this pull request to the merge queueSep 4, 2024
Merged via the queue into main with commit 7a24a01Sep 4, 2024
@jonrohan
jonrohan deleted the css_modules_link_staff branch September 4, 2024 15:24
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack