Skip to content

Bug fix: invisible Button variant missing background color when disabled - #4928

Merged
langermank merged 7 commits into
mainfrom
fix-invisible-disabled-bd
Sep 6, 2024
Merged

Bug fix: invisible Button variant missing background color when disabled#4928
langermank merged 7 commits into
mainfrom
fix-invisible-disabled-bd

Conversation

@langermank

Copy link
Copy Markdown
Contributor

Bug fix: invisible button variant missing bg color when disabled

@changeset-bot

changeset-botBot commented Sep 6, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 105426a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 6, 2024
@github-actions

github-actionsBot commented Sep 6, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.42 KB (-0.13% 🔽)
packages/react/dist/browser.umd.js96.77 KB (-0.06% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4928 September 6, 2024 16:45 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 6, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4928 September 6, 2024 17:04 Inactive
@langermank
langermank marked this pull request as ready for review September 6, 2024 17:56
@langermank
langermank requested a review from a team as a code ownerSeptember 6, 2024 17:56
cursor: not-allowed;
box-shadow: none;
color: primer.fg.disabled;
background-color: var(--button-invisible-bgColor-disabled,undefined);

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

What should the fallback value here be? Seems to be showing up as undefined 🤔

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Oh weird... lemme fix that

@langermank
langermank requested a review from a team as a code ownerSeptember 6, 2024 18:02
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 6, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4928 September 6, 2024 18:11 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 6, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4928 September 6, 2024 18:20 Inactive
@langermank
langermank added this pull request to the merge queueSep 6, 2024
Merged via the queue into main with commit caf4bcfSep 6, 2024
@langermank
langermank deleted the fix-invisible-disabled-bd branch September 6, 2024 18:44
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@langermank@jonrohan@joshblack