Skip to content

Button bug fixes: invisible variant icon colors missing variables + icon button disabled state - #4943

Merged
langermank merged 4 commits into
mainfrom
fix-invisible-icons
Sep 9, 2024
Merged

Button bug fixes: invisible variant icon colors missing variables + icon button disabled state#4943
langermank merged 4 commits into
mainfrom
fix-invisible-icons

Conversation

@langermank

Copy link
Copy Markdown
Contributor

The icons in the invisible button were just using fgColor-muted and missing the button specific variables. Disabled icon only buttons were missing disabled icon color.

@changeset-bot

changeset-botBot commented Sep 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9db31eb

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 9, 2024
@github-actions

github-actionsBot commented Sep 9, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.64 KB (+0.08% 🔺)
packages/react/dist/browser.umd.js96.91 KB (+0.04% 🔺)

@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 9, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4943 September 9, 2024 23:23 Inactive
@langermank
langermank marked this pull request as ready for review September 9, 2024 23:24
@langermank
langermank requested review from a team as code ownersSeptember 9, 2024 23:24
@langermank
langermank added this pull request to the merge queueSep 9, 2024
Merged via the queue into main with commit c282642Sep 9, 2024
@langermank
langermank deleted the fix-invisible-icons branch September 9, 2024 23:52
@primerprimerBot mentioned this pull request Sep 10, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@jonrohan