Skip to content

fix(Banner): update alignment of actions when Banner is dismissible - #4954

Merged
joshblack merged 7 commits into
mainfrom
fix/update-banner-actions-when-hidden-title
Sep 13, 2024
Merged

fix(Banner): update alignment of actions when Banner is dismissible#4954
joshblack merged 7 commits into
mainfrom
fix/update-banner-actions-when-hidden-title

Conversation

@joshblack

@joshblackjoshblack commented Sep 12, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3957

Changelog

New

Changed

  • Update styling for Banner to work when the banner is dismissible, has a hidden title, and actions

Removed

Rollout strategy

  • Patch release

Testing & Reviewing

  • Verify that the changes here address the issue above
  • Verify that no regressions occur in layout 😅

@changeset-bot

changeset-botBot commented Sep 12, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4b3a582

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 12, 2024
@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 12, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4954 September 12, 2024 16:29 Inactive
@github-actions

github-actionsBot commented Sep 12, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.16 KB (0%)
packages/react/dist/browser.umd.js97.4 KB (0%)

@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 12, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4954 September 12, 2024 16:37 Inactive
@joshblack
joshblack marked this pull request as ready for review September 12, 2024 16:57
@joshblack
joshblack requested review from a team as code ownersSeptember 12, 2024 16:57
@joshblack

Copy link
Copy Markdown
MemberAuthor

Note: just realized with the snapshots that the buttons are not in the right order, looking at this now 👀

@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 12, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4954 September 12, 2024 17:15 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 12, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4954 September 12, 2024 17:27 Inactive
Comment thread.changeset/spicy-eggs-melt.md Outdated
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4954 September 13, 2024 15:35 Inactive
@joshblack
joshblack added this pull request to the merge queueSep 13, 2024
Merged via the queue into main with commit af7f589Sep 13, 2024
@joshblack
joshblack deleted the fix/update-banner-actions-when-hidden-title branch September 13, 2024 15:52
@primerprimerBot mentioned this pull request Sep 13, 2024
TylerJDev pushed a commit that referenced this pull request Sep 23, 2024
…4954)
* fix(Banner): update alignment of actions when Banner is dismissible
* chore: add changeset
* test(vrt): update snapshots
* fix(Banner): update actions order
* test(vrt): update snapshots
* Update .changeset/spicy-eggs-melt.md
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@siddharthkp@langermank