Skip to content

feat: add experimental IssueLabel component - #4960

Merged
langermank merged 20 commits into
mainfrom
feat/add-label-component-2
Sep 25, 2024
Merged

feat: add experimental IssueLabel component#4960
langermank merged 20 commits into
mainfrom
feat/add-label-component-2

Conversation

@joshblack

@joshblackjoshblack commented Sep 13, 2024

Copy link
Copy Markdown
Member

Pulling work from #4535 to add support for an experimental IssueLabel component

This component mainly resembles the internal Label (Hubbers only). I removed props that seem to be unused/unnecessary, and also created better integration for "interactive" styles.

Changelog

New

  • Add experimental IssueLabel component

Changed

Removed

Rollout strategy

  • Minor release

joshblackand others added 6 commits September 13, 2024 11:09
* adding color algo for hex colors
* fix tests
* update snap name
* test(vrt): update snapshots
---------
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
Co-authored-by: langermank <langermank@users.noreply.github.com>
@changeset-bot

changeset-botBot commented Sep 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d3182c4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 13, 2024
@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4960 September 13, 2024 16:17 Inactive
@github-actions

github-actionsBot commented Sep 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.26 KB (0%)
packages/react/dist/browser.umd.js98.57 KB (0%)

@joshblackjoshblack added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Sep 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4960 September 13, 2024 18:15 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4960 September 13, 2024 18:26 Inactive
@joshblack
joshblack marked this pull request as ready for review September 13, 2024 18:54
@joshblack
joshblack requested review from a team as code ownersSeptember 13, 2024 18:54
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4960 September 24, 2024 22:46 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 24, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4960 September 24, 2024 23:01 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4960 September 24, 2024 23:51 Inactive
@langermanklangermank added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Sep 25, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 25, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4960 September 25, 2024 00:21 Inactive
@langermank
langermank marked this pull request as ready for review September 25, 2024 00:48
@langermank
langermank added this pull request to the merge queueSep 25, 2024
Merged via the queue into main with commit 21c3fceSep 25, 2024
@langermank
langermank deleted the feat/add-label-component-2 branch September 25, 2024 22:24
@primerprimerBot mentioned this pull request Sep 25, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@joshblack@lukasoppermann@broccolinisoup@langermank