Skip to content

refactor(Checkbox): update Checkbox to use CSS Modules behind flag - #4962

Merged
joshblack merged 15 commits into
mainfrom
3919-update-checkbox-css-modules-2
Sep 24, 2024
Merged

refactor(Checkbox): update Checkbox to use CSS Modules behind flag#4962
joshblack merged 15 commits into
mainfrom
3919-update-checkbox-css-modules-2

Conversation

@joshblack

@joshblackjoshblack commented Sep 13, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3919

Changelog

New

Changed

  • Update Checkbox to use CSS Modules behind flag

Removed

Rollout strategy

  • Minor release

@joshblack
joshblack requested a review from a team as a code ownerSeptember 13, 2024 19:38
@changeset-bot

changeset-botBot commented Sep 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2247fce

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4962 September 13, 2024 19:40 Inactive
@github-actions

github-actionsBot commented Sep 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.98 KB (+0.06% 🔺)
packages/react/dist/browser.umd.js98.3 KB (+0.05% 🔺)

@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4962 September 13, 2024 20:01 Inactive
@primer
primerBot requested a review from a team as a code ownerSeptember 13, 2024 20:08
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 13, 2024

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I think the focus color changes are just catching up with a previous primitives version bump

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4962 September 24, 2024 16:52 Inactive
@joshblack
joshblack added this pull request to the merge queueSep 24, 2024
Merged via the queue into main with commit 1977a68Sep 24, 2024
@joshblack
joshblack deleted the 3919-update-checkbox-css-modules-2 branch September 24, 2024 19:28
@primerprimerBot mentioned this pull request Sep 24, 2024
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/344079

@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@jonrohan@langermank