Skip to content

refactor(CounterLabel): update to use CSS Modules behind flag - #4965

Merged
joshblack merged 18 commits into
mainfrom
3918-update-counter-label
Sep 25, 2024
Merged

refactor(CounterLabel): update to use CSS Modules behind flag#4965
joshblack merged 18 commits into
mainfrom
3918-update-counter-label

Conversation

@joshblack

@joshblackjoshblack commented Sep 13, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3918

Changelog

New

Changed

  • Update CounterLabel to use a styled version under the hood instead of box to make it easier to migrate
  • Update CounterLabel to use CSS Modules
  • Refactor e2e tests for CounterLabel
  • Add additional story for CounterLabel

Removed

Rollout strategy

  • Minor release

@joshblack
joshblack requested a review from a team as a code ownerSeptember 13, 2024 20:46
@changeset-bot

changeset-botBot commented Sep 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d4119d9

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4965 September 13, 2024 20:48 Inactive
@github-actions

github-actionsBot commented Sep 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.31 KB (+0.05% 🔺)
packages/react/dist/browser.umd.js98.56 KB (-0.02% 🔽)

@primerprimer deleted a comment from primer-integrationBotSep 17, 2024
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/342701

Comment threadpackages/react/src/CounterLabel/CounterLabel.module.css Outdated
Comment threadpackages/react/src/CounterLabel/CounterLabel.module.css
joshblackand others added 2 commits September 20, 2024 18:38
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
@joshblackjoshblack added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Sep 24, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4965 September 24, 2024 18:30 Inactive
@joshblackjoshblack added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Sep 24, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4965 September 24, 2024 19:21 Inactive
@primer
primerBot requested a review from a team as a code ownerSeptember 24, 2024 19:29
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 24, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4965 September 24, 2024 19:31 Inactive
@joshblack
joshblack added this pull request to the merge queueSep 24, 2024
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Sep 24, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4965 September 25, 2024 15:22 Inactive
@joshblack
joshblack added this pull request to the merge queueSep 25, 2024
Merged via the queue into main with commit 5426a9aSep 25, 2024
@joshblack
joshblack deleted the 3918-update-counter-label branch September 25, 2024 15:43
@primerprimerBot mentioned this pull request Sep 25, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@jonrohan@langermank