Skip to content

SelectPanel: High contrast theme - #4966

Merged
siddharthkp merged 6 commits into
mainfrom
select-panel-high-contrast
Sep 19, 2024
Merged

SelectPanel: High contrast theme#4966
siddharthkp merged 6 commits into
mainfrom
select-panel-high-contrast

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Sep 16, 2024

Copy link
Copy Markdown
Member
Video 1Video 2
Before: Even though the cursor is moving and changing selection, you cannot see it.After: You can see the selection state and the selection being changed.
Screen.Recording.2024-09-16.at.4.05.12.PM.mov
Screen.Recording.2024-09-16.at.4.04.22.PM.mov

Rollout strategy

  • Patch release (with feature flag)
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Sep 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f28f75e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4966 September 16, 2024 11:22 Inactive
@github-actions

github-actionsBot commented Sep 16, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.52 KB (+0.1% 🔺)
packages/react/dist/browser.umd.js97.78 KB (+0.06% 🔺)

@siddharthkpsiddharthkp added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4966 September 16, 2024 14:01 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4966 September 16, 2024 14:11 Inactive
@siddharthkpsiddharthkp changed the title WIP: SelectPanel: High contrast themeSelectPanel: High contrast themeSep 16, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review September 16, 2024 14:14
@siddharthkp
siddharthkp requested review from a team as code ownersSeptember 16, 2024 14:14
@siddharthkpsiddharthkp self-assigned this Sep 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4966 September 16, 2024 14:15 Inactive

@siddharthkpsiddharthkpSep 17, 2024

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Note for reviewer: Our visual tests run on linux so this isn't exactly how it looks in windows. (see PR description for screenshots) but I think adding these to our snapshots will at least help us catch unexpected changes because we rarely test these manually

The light theme one is surprisingly similar, but the dark theme is just very different

@camertroncamertron left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nice, thanks! 🚀

@siddharthkp
siddharthkp added this pull request to the merge queueSep 19, 2024
Merged via the queue into main with commit c097e96Sep 19, 2024
@siddharthkp
siddharthkp deleted the select-panel-high-contrast branch September 19, 2024 13:40
@primerprimerBot mentioned this pull request Sep 19, 2024
TylerJDev pushed a commit that referenced this pull request Sep 23, 2024
* use border-width to make icon visible
* ActionList: include active-descendant for high contrast
* e2e test for forced-colors (windows high contrast)
* test(vrt): update snapshots
* Create ten-gifts-own.md
* update snapshots
---------
Co-authored-by: siddharthkp <siddharthkp@users.noreply.github.com>
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@camertron