Skip to content

Moving Heading feature flag from staff ship to ga ship - #4971

Merged
jonrohan merged 2 commits into
mainfrom
css_module_ga_heading
Sep 17, 2024
Merged

Moving Heading feature flag from staff ship to ga ship#4971
jonrohan merged 2 commits into
mainfrom
css_module_ga_heading

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Part of https://github.com/github/primer/issues/3716

Changelog

Changed

The feature flag for Heading changed from primer_react_css_modules_staff to primer_react_css_modules_ga

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

@jonrohan
jonrohan requested a review from a team as a code ownerSeptember 16, 2024 19:16
@changeset-bot

changeset-botBot commented Sep 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 274deaa

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 16, 2024
@github-actions

github-actionsBot commented Sep 16, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.5 KB (+0.03% 🔺)
packages/react/dist/browser.umd.js97.65 KB (-0.16% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4971 September 16, 2024 19:19 Inactive
@jonrohanjonrohan changed the title Moving Heading feature flag from staff ship to team shipMoving Heading feature flag from staff ship to ga shipSep 16, 2024

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Yay exciting!

@jonrohan
jonrohan added this pull request to the merge queueSep 17, 2024
Merged via the queue into main with commit ff56b04Sep 17, 2024
@jonrohan
jonrohan deleted the css_module_ga_heading branch September 17, 2024 16:39
@primerprimerBot mentioned this pull request Sep 17, 2024
TylerJDev pushed a commit that referenced this pull request Sep 23, 2024
* Moving Heading component to ga flag
* Create nice-moles-know.md
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@siddharthkp