Skip to content

fix(FormControl): update Caption to no longer use styled system props - #4972

Merged
joshblack merged 3 commits into
mainfrom
fix/update-text-checkbox-tests
Sep 16, 2024
Merged

fix(FormControl): update Caption to no longer use styled system props#4972
joshblack merged 3 commits into
mainfrom
fix/update-text-checkbox-tests

Conversation

@joshblack

Copy link
Copy Markdown
Member

Update our internal component for captions to no longer use styled system props. This caused an issue when our CSS Module flags were enabled since we were no longer forwarding these props along to Box to correctly generate class names

Changelog

New

Changed

  • Update how our internal InputCaption component styles the Text component so that it no longer uses styled-system props

Removed

Rollout strategy

  • Patch release

Testing & Reviewing

  • Verify that when flag is enabled that the Checkbox With Caption story renders as expected

@joshblack
joshblack requested a review from a team as a code ownerSeptember 16, 2024 19:56
@changeset-bot

changeset-botBot commented Sep 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6b993e1

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 16, 2024
@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4972 September 16, 2024 19:59 Inactive
@github-actions

github-actionsBot commented Sep 16, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.46 KB (-0.01% 🔽)
packages/react/dist/browser.umd.js97.72 KB (-0.09% 🔽)

@primer
primerBot requested a review from a team as a code ownerSeptember 16, 2024 20:26
@primer
primerBot requested a review from tbenningSeptember 16, 2024 20:26
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4972 September 16, 2024 20:29 Inactive
@joshblack
joshblack added this pull request to the merge queueSep 16, 2024
Merged via the queue into main with commit 082b4e7Sep 16, 2024
@joshblack
joshblack deleted the fix/update-text-checkbox-tests branch September 16, 2024 22:07
@primerprimerBot mentioned this pull request Sep 16, 2024
TylerJDev pushed a commit that referenced this pull request Sep 23, 2024
…#4972)
* fix(FormControl): update Caption to no longer use styled system props
* chore: add changeset
* test(vrt): update snapshots
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@langermank