Skip to content

Change the Link component feature flag from primer_react_css_modules_staff to primer_react_css_modules_ga - #4975

Merged
jonrohan merged 3 commits into
mainfrom
css_modules_ga_link
Sep 18, 2024
Merged

Change the Link component feature flag from primer_react_css_modules_staff to primer_react_css_modules_ga#4975
jonrohan merged 3 commits into
mainfrom
css_modules_ga_link

Conversation

@jonrohan

@jonrohanjonrohan commented Sep 16, 2024

Copy link
Copy Markdown
Member

Part of https://github.com/github/primer/issues/3719

Changelog

Changed

Change the Link component feature flag from primer_react_css_modules_staff to primer_react_css_modules_ga

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

@changeset-bot

changeset-botBot commented Sep 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 728447c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4975 September 16, 2024 20:35 Inactive
@github-actions

github-actionsBot commented Sep 16, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.4 KB (-0.11% 🔽)
packages/react/dist/browser.umd.js97.67 KB (+0.01% 🔺)

@jonrohan
jonrohan marked this pull request as ready for review September 16, 2024 20:38
@jonrohan
jonrohan requested a review from a team as a code ownerSeptember 16, 2024 20:38
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4975 September 16, 2024 20:39 Inactive
@jonrohan
jonrohan added this pull request to the merge queueSep 18, 2024
Merged via the queue into main with commit 0b89fc0Sep 18, 2024
@jonrohan
jonrohan deleted the css_modules_ga_link branch September 18, 2024 16:31
@primerprimerBot mentioned this pull request Sep 18, 2024
TylerJDev pushed a commit that referenced this pull request Sep 23, 2024
…staff to primer_react_css_modules_ga (#4975)
* Move Link to ga feature flag
* Create dirty-pianos-wash.md
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack