Skip to content

Removing the CSS modules feature flag from Blankslate - #4981

Merged
jonrohan merged 2 commits into
mainfrom
remove_blankslate_css_modules_feature_flag
Sep 17, 2024
Merged

Removing the CSS modules feature flag from Blankslate#4981
jonrohan merged 2 commits into
mainfrom
remove_blankslate_css_modules_feature_flag

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3696

Changelog

Removed

Removes the feature flag from the Blankslate component to always render with CSS modules.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Merge checklist

@changeset-bot

changeset-botBot commented Sep 17, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d98bf21

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 17, 2024
@github-actions

github-actionsBot commented Sep 17, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.58 KB (0%)
packages/react/dist/browser.umd.js97.7 KB (0%)

@jonrohan
jonrohan marked this pull request as ready for review September 17, 2024 17:44
@jonrohan
jonrohan requested a review from a team as a code ownerSeptember 17, 2024 17:44
@jonrohan
jonrohan added this pull request to the merge queueSep 17, 2024
Merged via the queue into main with commit c8fe1c6Sep 17, 2024
@jonrohan
jonrohan deleted the remove_blankslate_css_modules_feature_flag branch September 17, 2024 21:20
@primerprimerBot mentioned this pull request Sep 17, 2024
TylerJDev pushed a commit that referenced this pull request Sep 23, 2024
* Removing the css modules feature flag from blankslate
* Create dull-turkeys-cheer.md
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack