Skip to content

feat(react): remove deprecated components from main entrypoint - #4992

Merged
joshblack merged 8 commits into
mainfrom
feat/remove-deprecated-from-deprecated
Sep 30, 2024
Merged

feat(react): remove deprecated components from main entrypoint#4992
joshblack merged 8 commits into
mainfrom
feat/remove-deprecated-from-deprecated

Conversation

@joshblack

@joshblackjoshblack commented Sep 19, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4012
Closeshttps://github.com/github/primer/issues/4013
Closeshttps://github.com/github/primer/issues/4014
Closeshttps://github.com/github/primer/issues/4015

With https://github.com/github/github/pull/343108 merged, and the lint rule enabled, we can now remove deprecated components from the main entrypoint.

Changelog

New

Changed

  • Move the following components to deprecated:
    • Octicon
    • Pagehead
    • Tooltip (v1)
    • Dialog (v1)
  • Update tests, docs, and storybook entries to indicate that they are deprecated

Removed

Rollout strategy

  • Major release; if selected, include a written rollout or migration plan

We have created a rule to forbid importing these components from @primer/react and have migrated them to @primer/react/deprecated upstream and enabled that rule. Removing these exports should have no effect as a result as all instances of imports have been removed.

@changeset-bot

changeset-botBot commented Sep 19, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9060a98

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 19, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4992 September 19, 2024 19:14 Inactive
@github-actions

github-actionsBot commented Sep 19, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.74 KB (-1.39% 🔽)
packages/react/dist/browser.umd.js97.13 KB (-1.24% 🔽)

@joshblack
joshblack marked this pull request as ready for review September 24, 2024 17:06
@joshblack
joshblack requested a review from a team as a code ownerSeptember 24, 2024 17:06
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4992 September 24, 2024 17:09 Inactive

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Changes look good 👍🏻

@joshblack
joshblack added this pull request to the merge queueSep 24, 2024
@joshblack
joshblack removed this pull request from the merge queue due to a manual request Sep 24, 2024

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I see you tested it with dotcom, that's the main thing here :)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/344709

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4992 September 30, 2024 15:50 Inactive
@joshblack
joshblack added this pull request to the merge queueSep 30, 2024
Merged via the queue into main with commit d1d911aSep 30, 2024
@joshblack
joshblack deleted the feat/remove-deprecated-from-deprecated branch September 30, 2024 16:57
@primerprimerBot mentioned this pull request Sep 30, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@jonrohan@siddharthkp