Skip to content

Moving Label and Text to primer_react_css_modules_staff ship feature flag - #5000

Merged
jonrohan merged 3 commits into
mainfrom
move_label_text_to_staff_ship
Sep 20, 2024
Merged

Moving Label and Text to primer_react_css_modules_staff ship feature flag#5000
jonrohan merged 3 commits into
mainfrom
move_label_text_to_staff_ship

Conversation

@jonrohan

@jonrohanjonrohan commented Sep 20, 2024

Copy link
Copy Markdown
Member

Part of https://github.com/github/primer/issues/3785
Part of https://github.com/github/primer/issues/3914

Changelog

Changed

  • Label component CSS module feature flag changed to primer_react_css_modules_staff
  • Text component CSS module feature flag changed to primer_react_css_modules_staff

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

@changeset-bot

changeset-botBot commented Sep 20, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0183655

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 20, 2024
@jonrohan
jonrohan marked this pull request as ready for review September 20, 2024 18:39
@jonrohan
jonrohan requested a review from a team as a code ownerSeptember 20, 2024 18:39
@github-actions

github-actionsBot commented Sep 20, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.84 KB (+0.06% 🔺)
packages/react/dist/browser.umd.js98.07 KB (-0.11% 🔽)

@jonrohan
jonrohan added this pull request to the merge queueSep 20, 2024
Merged via the queue into main with commit b8f6888Sep 20, 2024
@jonrohan
jonrohan deleted the move_label_text_to_staff_ship branch September 20, 2024 21:02
@primerprimerBot mentioned this pull request Sep 20, 2024
TylerJDev pushed a commit that referenced this pull request Sep 23, 2024
…flag (#5000)
* Moving Label and Text to staff ship feature flag
* Create serious-sheep-love.md
* Create big-pumas-sit.md
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack