Skip to content

chore(ButtonGroup): Migrate ButtonGroup component to use CSS modules behind the primer_react_css_modules_team feature flag - #5001

Merged
jonrohan merged 5 commits into
mainfrom
css_modules_button_group
Sep 23, 2024
Merged

chore(ButtonGroup): Migrate ButtonGroup component to use CSS modules behind the primer_react_css_modules_team feature flag#5001
jonrohan merged 5 commits into
mainfrom
css_modules_button_group

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3960

Changelog

Changed

Migrate ButtonGroup component to use CSS modules behind the primer_react_css_modules_team feature flag

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

@changeset-bot

changeset-botBot commented Sep 20, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0541d46

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 20, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5001 September 20, 2024 19:44 Inactive
@github-actions

github-actionsBot commented Sep 20, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.03 KB (+0.26% 🔺)
packages/react/dist/browser.umd.js98.17 KB (-0.01% 🔽)

@jonrohan
jonrohan marked this pull request as ready for review September 20, 2024 19:47
@jonrohan
jonrohan requested a review from a team as a code ownerSeptember 20, 2024 19:47
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5001 September 20, 2024 19:49 Inactive

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🥳

@jonrohan
jonrohan added this pull request to the merge queueSep 23, 2024
Merged via the queue into main with commit 597d285Sep 23, 2024
@jonrohan
jonrohan deleted the css_modules_button_group branch September 23, 2024 16:28
@primerprimerBot mentioned this pull request Sep 23, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@broccolinisoup