Skip to content

chore(AvatarPair): Convert AvatarPair to CSS modules - #5002

Merged
jonrohan merged 4 commits into
mainfrom
css_modules_avatarpair
Sep 23, 2024
Merged

chore(AvatarPair): Convert AvatarPair to CSS modules#5002
jonrohan merged 4 commits into
mainfrom
css_modules_avatarpair

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4023

I opted in this case to go directly to CSS modules as this component will be removed in the next major release.

Changelog

Changed

Converts AvatarPair to CSS modules

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerSeptember 20, 2024 21:06
@changeset-bot

changeset-botBot commented Sep 20, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: abc9f25

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 20, 2024
@github-actions

github-actionsBot commented Sep 20, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.96 KB (-0.03% 🔽)
packages/react/dist/browser.umd.js98.23 KB (-0.05% 🔽)

@jonrohanjonrohan added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 20, 2024
@primer
primerBot requested a review from a team as a code ownerSeptember 20, 2024 21:34
@primer
primerBot requested a review from langermankSeptember 20, 2024 21:34
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 20, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5002 September 20, 2024 21:36 Inactive
@jonrohanjonrohan added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 23, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 23, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5002 September 23, 2024 16:48 Inactive
@jonrohan
jonrohan added this pull request to the merge queueSep 23, 2024
Merged via the queue into main with commit 9bd5c89Sep 23, 2024
@jonrohan
jonrohan deleted the css_modules_avatarpair branch September 23, 2024 17:05
@primerprimerBot mentioned this pull request Sep 23, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack