Skip to content

Add bg prop to ProgressBar.Item - #5007

Merged
lukasoppermann merged 3 commits into
mainfrom
@lukasoppermann/progressbar
Sep 24, 2024
Merged

Add bg prop to ProgressBar.Item#5007
lukasoppermann merged 3 commits into
mainfrom
@lukasoppermann/progressbar

Conversation

@lukasoppermann

@lukasoppermannlukasoppermann commented Sep 23, 2024

Copy link
Copy Markdown
Contributor

Adds a bg prop for on ProgressBar.Item as discussed in https://github.slack.com/archives/GACAW0NPM/p1726839001767409

Follow-up task

Changelog

  • Adds a bg prop for on ProgressBar.Item

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Sep 23, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2ef06ac

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

lukasoppermann added a commit to primer/eslint-plugin-primer-react that referenced this pull request Sep 23, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5007 September 23, 2024 07:35 Inactive
@lukasoppermann
lukasoppermannforce-pushed the @lukasoppermann/progressbar branch from 71530f0 to 09c95e3CompareSeptember 23, 2024 07:37
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5007 September 23, 2024 07:39 Inactive
@github-actions

github-actionsBot commented Sep 23, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.84 KB (-0.13% 🔽)
packages/react/dist/browser.umd.js98.21 KB (-0.03% 🔽)

@lukasoppermann
lukasoppermannforce-pushed the @lukasoppermann/progressbar branch from 623c2de to 0a88699CompareSeptember 23, 2024 08:29
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5007 September 23, 2024 08:31 Inactive
lukasoppermann added a commit to primer/eslint-plugin-primer-react that referenced this pull request Sep 23, 2024
* Add bg for ProgressBar.item
Merge this once primer/react#5007 is merged
* Create beige-schools-horse.md

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Request some minor improvements

Comment thread.changeset/slow-walls-drum.md Outdated
Comment threadpackages/react/src/ProgressBar/ProgressBar.tsx
Co-authored-by: Siddharth Kshetrapal <siddharthkp@github.com>
if (args.bg && args.bg !== '') {
setSectionColors([args.bg, ...sectionColorsDefault])
}
}, [args.bg])

@siddharthkpsiddharthkpSep 24, 2024

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

(non blocker, nit picking) Didn't really understand what this block of code does, can you help me?

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Needed to avoid error: "To many re-renders"

@lukasoppermann
lukasoppermann added this pull request to the merge queueSep 24, 2024
Merged via the queue into main with commit c909285Sep 24, 2024
@lukasoppermann
lukasoppermann deleted the @lukasoppermann/progressbar branch September 24, 2024 11:16
@primerprimerBot mentioned this pull request Sep 24, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
WilliamSmith1226 added a commit to WilliamSmith1226/React-eslint-plugin-primer-components that referenced this pull request Jan 6, 2026
* Add bg for ProgressBar.item
Merge this once primer/react#5007 is merged
* Create beige-schools-horse.md
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@lukasoppermann@siddharthkp