Skip to content

feat: promote Tooltip, Dialog, Stack to stable - #5010

Merged
broccolinisoup merged 12 commits into
mainfrom
feat/promote-tooltip-dialog-stack
Oct 8, 2024
Merged

feat: promote Tooltip, Dialog, Stack to stable#5010
broccolinisoup merged 12 commits into
mainfrom
feat/promote-tooltip-dialog-stack

Conversation

@joshblack

@joshblackjoshblack commented Sep 23, 2024

Copy link
Copy Markdown
Member

Add Tooltip (V2), Dialog (V2), and Stack to stable entrypoint. This is a part of our v37 release.

Changelog

New

Changed

  • Add Tooltip V2 to @primer/react
  • Add Dialog V2 to @primer/react
  • Add Stack to @primer/react Apparently this was already done 😂

Removed

Rollout strategy

  • Minor release

We will remove these components from @primer/react/next and @primer/react/experimental once dotcom is migrated

@changeset-bot

changeset-botBot commented Sep 23, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9880ade

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 23, 2024
@github-actions

github-actionsBot commented Sep 23, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.73 KB (+0.03% 🔺)
packages/react/dist/browser.umd.js96.96 KB (-0.06% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5010 October 3, 2024 14:50 Inactive
@joshblack
joshblack marked this pull request as ready for review October 3, 2024 14:58
@joshblack
joshblack requested a review from a team as a code ownerOctober 3, 2024 14:58
@joshblack

Copy link
Copy Markdown
MemberAuthor

@broccolinisoup could this go out in the next release, if possible? 🙏

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5010 October 7, 2024 23:41 Inactive
@broccolinisoup
broccolinisoup added this pull request to the merge queueOct 8, 2024
Merged via the queue into main with commit 8385c33Oct 8, 2024
@broccolinisoup
broccolinisoup deleted the feat/promote-tooltip-dialog-stack branch October 8, 2024 00:36
@primerprimerBot mentioned this pull request Oct 8, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@broccolinisoup