Skip to content

Bug fix: Button aria-expanded state (CSS Modules) - #5011

Merged
langermank merged 7 commits into
mainfrom
button-expanded-bugfix
Sep 23, 2024
Merged

Bug fix: Button aria-expanded state (CSS Modules)#5011
langermank merged 7 commits into
mainfrom
button-expanded-bugfix

Conversation

@langermank

Copy link
Copy Markdown
Contributor

Bug fix for CSS Modules: aria-expanded should have a hover state

@langermank
langermank requested a review from a team as a code ownerSeptember 23, 2024 20:19
@changeset-bot

changeset-botBot commented Sep 23, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ff828a8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 23, 2024
@github-actions

github-actionsBot commented Sep 23, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.96 KB (0%)
packages/react/dist/browser.umd.js98.23 KB (0%)

@primer
primerBot requested a review from a team as a code ownerSeptember 23, 2024 20:27
@primer
primerBot requested a review from emilybrickSeptember 23, 2024 20:27
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 23, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5011 September 23, 2024 20:29 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5011 September 23, 2024 20:47 Inactive
@langermank
langermank added this pull request to the merge queueSep 23, 2024
Merged via the queue into main with commit 7a5205dSep 23, 2024
@langermank
langermank deleted the button-expanded-bugfix branch September 23, 2024 21:09
@primerprimerBot mentioned this pull request Sep 23, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@joshblack