Skip to content

feat: update stable and deprecated exports - #5021

Merged
joshblack merged 3 commits into
mainfrom
feat/update-entry-points-exports-for-responsive-value
Oct 1, 2024
Merged

feat: update stable and deprecated exports#5021
joshblack merged 3 commits into
mainfrom
feat/update-entry-points-exports-for-responsive-value

Conversation

@joshblack

Copy link
Copy Markdown
Member

Adds a missing ResponsiveValue export and remove ForwardRefComponent from deprecated since we'll just update this upstream in dotcom 👀

Changelog

New

Changed

  • Add ResponsiveValue to @primer/react

Removed

  • Remove ForwardRefComponent from @primer/react/deprecated (this hasn't made it's way to a release yet)

@joshblack
joshblack requested a review from a team as a code ownerSeptember 24, 2024 21:57
@changeset-bot

changeset-botBot commented Sep 24, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ddf408f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 24, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5021 September 24, 2024 21:59 Inactive
@github-actions

github-actionsBot commented Sep 24, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.74 KB (0%)
packages/react/dist/browser.umd.js97.13 KB (0%)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/344938

@joshblack
joshblack added this pull request to the merge queueOct 1, 2024
Merged via the queue into main with commit cff067aOct 1, 2024
@joshblack
joshblack deleted the feat/update-entry-points-exports-for-responsive-value branch October 1, 2024 16:14
@primerprimerBot mentioned this pull request Oct 1, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@TylerJDev