Skip to content

Move Avatar component feature flag from primer_react_css_modules_team to primer_react_css_modules_staff - #5024

Merged
jonrohan merged 5 commits into
mainfrom
css_modules_avatar_to_staff
Sep 26, 2024
Merged

Move Avatar component feature flag from primer_react_css_modules_team to primer_react_css_modules_staff#5024
jonrohan merged 5 commits into
mainfrom
css_modules_avatar_to_staff

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Changelog

Changed

Move Avatar component feature flag from primer_react_css_modules_team to primer_react_css_modules_staff

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Sep 24, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b20477f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 24, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5024 September 24, 2024 22:11 Inactive
@github-actions

github-actionsBot commented Sep 24, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.09 KB (-0.12% 🔽)
packages/react/dist/browser.umd.js98.34 KB (-0.08% 🔽)

@jonrohan
jonrohan marked this pull request as ready for review September 24, 2024 22:19
@jonrohan
jonrohan requested a review from a team as a code ownerSeptember 24, 2024 22:19
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5024 September 26, 2024 22:11 Inactive
@jonrohan
jonrohan added this pull request to the merge queueSep 26, 2024
Merged via the queue into main with commit 45f3597Sep 26, 2024
@jonrohan
jonrohan deleted the css_modules_avatar_to_staff branch September 26, 2024 22:31
@primerprimerBot mentioned this pull request Sep 26, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack