Skip to content

Banner: Fix alignment of secondary action without primary action - #5026

Merged
siddharthkp merged 4 commits into
mainfrom
banner/fix-secondary-action-alignment
Sep 25, 2024
Merged

Banner: Fix alignment of secondary action without primary action#5026
siddharthkp merged 4 commits into
mainfrom
banner/fix-secondary-action-alignment

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Sep 25, 2024

Copy link
Copy Markdown
Member
BeforeAfter
After: secondary button and dismiss button are alignedBefore: secondary button is misaligned with dismiss button

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@siddharthkpsiddharthkp self-assigned this Sep 25, 2024
@changeset-bot

changeset-botBot commented Sep 25, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 97a8a85

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 25, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review September 25, 2024 14:27
@siddharthkp
siddharthkp requested a review from a team as a code ownerSeptember 25, 2024 14:27
@siddharthkpsiddharthkp added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 25, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5026 September 25, 2024 14:28 Inactive
@github-actions

github-actionsBot commented Sep 25, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.26 KB (0%)
packages/react/dist/browser.umd.js98.57 KB (0%)

@primer
primerBot requested a review from a team as a code ownerSeptember 25, 2024 14:35
@primer
primerBot requested a review from emilybrickSeptember 25, 2024 14:35
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 25, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5026 September 25, 2024 14:38 Inactive

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thank you so much 🙏

@siddharthkpsiddharthkp changed the title Banner: Fix alignment of secondary button on it's ownBanner: Fix alignment of secondary action without primary actionSep 25, 2024
@joshblack

Copy link
Copy Markdown
Member

I wonder what's going on in the snapshots 🤔 it seems to look correct locally and in the checked in screenshots

@siddharthkpsiddharthkp added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 25, 2024
@siddharthkp

Copy link
Copy Markdown
MemberAuthor

I wonder what's going on in the snapshots 🤔 it seems to look correct locally and in the checked in screenshots

Yeah kinda weird, given that the snapshots were just added 😓 Running it once more

@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 25, 2024
@siddharthkpsiddharthkp added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 25, 2024
@siddharthkp

siddharthkp commented Sep 25, 2024

Copy link
Copy Markdown
MemberAuthor

@joshblack figured it out, didn't realise Banner has 2 places with css now!! 😅

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5026 September 25, 2024 15:28 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 25, 2024
@siddharthkp
siddharthkp added this pull request to the merge queueSep 25, 2024
Merged via the queue into main with commit 3302440Sep 25, 2024
@siddharthkp
siddharthkp deleted the banner/fix-secondary-action-alignment branch September 25, 2024 15:47
@primerprimerBot mentioned this pull request Sep 25, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@joshblack