Skip to content

refactor(BranchName): update BranchName to use CSS Modules - #5028

Merged
joshblack merged 5 commits into
mainfrom
refactor/update-branch-name-to-css-modules
Sep 26, 2024
Merged

refactor(BranchName): update BranchName to use CSS Modules#5028
joshblack merged 5 commits into
mainfrom
refactor/update-branch-name-to-css-modules

Conversation

@joshblack

@joshblackjoshblack commented Sep 25, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4040

Changelog

New

Changed

  • Update BranchName to use CSS Modules behind a feature flag

Removed

Rollout strategy

  • Minor release

@changeset-bot

changeset-botBot commented Sep 25, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1f46f2d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 25, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5028 September 25, 2024 18:16 Inactive
@github-actions

github-actionsBot commented Sep 25, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.2 KB (+0.26% 🔺)
packages/react/dist/browser.umd.js98.42 KB (+0.02% 🔺)

@joshblack
joshblack marked this pull request as ready for review September 25, 2024 18:46
@joshblack
joshblack requested a review from a team as a code ownerSeptember 25, 2024 18:46
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5028 September 25, 2024 18:47 Inactive
@primer-integration

Copy link
Copy Markdown

😢 Hi from github/github. The integration workflow has failed: https://github.com/github/github/actions/runs/11039278345

Comment threadpackages/react/src/BranchName/BranchName.tsx
@joshblack
joshblack added this pull request to the merge queueSep 26, 2024
Merged via the queue into main with commit c2dd024Sep 26, 2024
@joshblack
joshblack deleted the refactor/update-branch-name-to-css-modules branch September 26, 2024 22:02
@primerprimerBot mentioned this pull request Sep 26, 2024
@joshblack
joshblack restored the refactor/update-branch-name-to-css-modules branch September 27, 2024 14:11
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@jonrohan