Skip to content

refactor(BranchName): update BranchName to use CSS Modules - #5040

Merged
joshblack merged 15 commits into
mainfrom
refactor/update-branch-name-to-css-modules
Oct 31, 2024
Merged

refactor(BranchName): update BranchName to use CSS Modules#5040
joshblack merged 15 commits into
mainfrom
refactor/update-branch-name-to-css-modules

Conversation

@joshblack

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4040

Changelog

New

Changed

  • Update BranchName to use CSS Modules behind a feature flag

Removed

Rollout strategy

  • Minor release

@joshblack
joshblack requested a review from a team as a code ownerSeptember 27, 2024 14:22
@changeset-bot

changeset-botBot commented Sep 27, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: faadbfe

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Sep 27, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5040 September 27, 2024 14:24 Inactive
@github-actions

github-actionsBot commented Sep 27, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.55 KB (+0.17% 🔺)
packages/react/dist/browser.umd.js97.88 KB (+0.13% 🔺)

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 23, 2024
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/349287

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5040 October 31, 2024 18:03 Inactive
@joshblack
joshblack added this pull request to the merge queueOct 31, 2024
Merged via the queue into main with commit 8d9a357Oct 31, 2024
@joshblack
joshblack deleted the refactor/update-branch-name-to-css-modules branch October 31, 2024 18:17
@primerprimerBot mentioned this pull request Oct 31, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@jonrohan