Skip to content

Banner: Fix margin for inline actions - #5053

Merged
siddharthkp merged 4 commits into
mainfrom
banner-fix-actions-alignment-with-dismissible
Oct 1, 2024
Merged

Banner: Fix margin for inline actions#5053
siddharthkp merged 4 commits into
mainfrom
banner-fix-actions-alignment-with-dismissible

Conversation

@siddharthkp

Copy link
Copy Markdown
Member

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@siddharthkpsiddharthkp added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Oct 1, 2024
@changeset-bot

changeset-botBot commented Oct 1, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c1fb589

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 1, 2024
@siddharthkpsiddharthkp changed the title we don't need margin when actions are inlineBanner: Fix margin for inline actionsOct 1, 2024
@siddharthkpsiddharthkp self-assigned this Oct 1, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review October 1, 2024 11:34
@siddharthkp
siddharthkp requested a review from a team as a code ownerOctober 1, 2024 11:34
@github-actions

github-actionsBot commented Oct 1, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.74 KB (0%)
packages/react/dist/browser.umd.js97.13 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5053 October 1, 2024 11:36 Inactive
@primer
primerBot requested a review from a team as a code ownerOctober 1, 2024 11:44
@primer
primerBot requested a review from mperrottiOctober 1, 2024 11:44
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Oct 1, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5053 October 1, 2024 11:46 Inactive
@siddharthkp
siddharthkp added this pull request to the merge queueOct 1, 2024
Merged via the queue into main with commit 2703d0cOct 1, 2024
@siddharthkp
siddharthkp deleted the banner-fix-actions-alignment-with-dismissible branch October 1, 2024 16:45
@primerprimerBot mentioned this pull request Oct 1, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Spacing is off when using Banner with hideTitle, actions, and dismiss button

2 participants

@siddharthkp@joshblack