Skip to content

chore(Avatar): Move CSS modules feature flag from staff to ga for Avatar component - #5062

Merged
jonrohan merged 3 commits into
mainfrom
css_modules_avatar_ga
Oct 8, 2024
Merged

chore(Avatar): Move CSS modules feature flag from staff to ga for Avatar component#5062
jonrohan merged 3 commits into
mainfrom
css_modules_avatar_ga

Conversation

@jonrohan

@jonrohanjonrohan commented Oct 4, 2024

Copy link
Copy Markdown
Member

Changelog

Changed

This moves the CSS modules feature flag from staff to ga for the Avatar component.

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Merge checklist

@changeset-bot

changeset-botBot commented Oct 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 90e0a08

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5062 October 4, 2024 00:03 Inactive
@github-actions

github-actionsBot commented Oct 4, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.75 KB (+0.06% 🔺)
packages/react/dist/browser.umd.js96.98 KB (-0.05% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5062 October 4, 2024 00:06 Inactive
@jonrohan
jonrohan marked this pull request as ready for review October 7, 2024 20:39
@jonrohan
jonrohan requested a review from a team as a code ownerOctober 7, 2024 20:39
@jonrohan
jonrohan added this pull request to the merge queueOct 8, 2024
Merged via the queue into main with commit 0f5d5e0Oct 8, 2024
@jonrohan
jonrohan deleted the css_modules_avatar_ga branch October 8, 2024 00:19
@primerprimerBot mentioned this pull request Oct 8, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@broccolinisoup