Skip to content

bugFix: When the feature flag is enabled and sx prop is passed in use, Box - #5068

Merged
jonrohan merged 4 commits into
mainfrom
toggle_styled_component_sx_prop_fix
Oct 4, 2024
Merged

bugFix: When the feature flag is enabled and sx prop is passed in use, Box#5068
jonrohan merged 4 commits into
mainfrom
toggle_styled_component_sx_prop_fix

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Fixes an issue when the flag is enabled and an sx prop is passed in, we want to still process the sx.

Changelog

Changed

When the feature flag is enabled and sx prop is passed in use the Box component

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerOctober 4, 2024 18:07
@changeset-bot

changeset-botBot commented Oct 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a4ade5a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 4, 2024
@github-actions

github-actionsBot commented Oct 4, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.76 KB (+0.04% 🔺)
packages/react/dist/browser.umd.js97.08 KB (+0.08% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5068 October 4, 2024 18:10 Inactive
Comment threadpackages/react/src/internal/utils/toggleStyledComponent.tsx
@jonrohan
jonrohan added this pull request to the merge queueOct 4, 2024
Merged via the queue into main with commit 5f7bd0bOct 4, 2024
@jonrohan
jonrohan deleted the toggle_styled_component_sx_prop_fix branch October 4, 2024 20:19
@primerprimerBot mentioned this pull request Oct 4, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack