Skip to content

feat(react): add useFeatureFlag to experimental entrypoint - #5074

Merged
joshblack merged 2 commits into
mainfrom
feat/add-missing-imports-2
Oct 7, 2024
Merged

feat(react): add useFeatureFlag to experimental entrypoint#5074
joshblack merged 2 commits into
mainfrom
feat/add-missing-imports-2

Conversation

@joshblack

Copy link
Copy Markdown
Member

This is a follow-up to removing wildcard entrypoints from Primer React. In the latest version of dotcom, we are seeing usage of useFeatureFlag that we are now adding to the experimental entrypoint.

Changelog

New

  • Add useFeatureFlag to the experimental entrypoint

Changed

Removed

Rollout strategy

  • Minor release

@joshblack
joshblack requested a review from a team as a code ownerOctober 7, 2024 15:32
@changeset-bot

changeset-botBot commented Oct 7, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: dad9bc8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 7, 2024
@joshblack

Copy link
Copy Markdown
MemberAuthor

@broccolinisoup would it be possible if this could go out in the next release? 🙏

@github-actions

github-actionsBot commented Oct 7, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.93 KB (0%)
packages/react/dist/browser.umd.js97.12 KB (0%)

@joshblack
joshblack added this pull request to the merge queueOct 7, 2024
Merged via the queue into main with commit e4965edOct 7, 2024
@joshblack
joshblack deleted the feat/add-missing-imports-2 branch October 7, 2024 18:54
@primerprimerBot mentioned this pull request Oct 7, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@jonrohan