Skip to content

refactor(react): remove wildcard imports - #5075

Merged
joshblack merged 5 commits into
mainfrom
3688-remove-wildcard-exports
Oct 18, 2024
Merged

refactor(react): remove wildcard imports#5075
joshblack merged 5 commits into
mainfrom
3688-remove-wildcard-exports

Conversation

@joshblack

@joshblackjoshblack commented Oct 7, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3688

Update our package in @primer/react to no longer allow wildcard imports. This will allow us to explicitly define files that folks can import from @primer/react and means that we can avoid our internal convention in the future.

Changelog

New

Changed

Removed

  • Remove support for wildcard imports from our package

Rollout strategy

  • Major release; if selected, include a written rollout or migration plan

We have added wildcard imports that are used to our named entrypoints. As a result, we are going to update imports upstream to use these instead of named entrypoints.

@changeset-bot

changeset-botBot commented Oct 7, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a94f4f7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 7, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5075 October 7, 2024 16:22 Inactive
@github-actions

github-actionsBot commented Oct 7, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.06 KB (0%)
packages/react/dist/browser.umd.js97.36 KB (0%)

@joshblack
joshblack marked this pull request as ready for review October 9, 2024 20:01
@joshblack
joshblack requested a review from a team as a code ownerOctober 9, 2024 20:01
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 9, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5075 October 9, 2024 20:05 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5075 October 17, 2024 21:51 Inactive
@joshblack
joshblack added this pull request to the merge queueOct 18, 2024
Merged via the queue into main with commit de7dbdeOct 18, 2024
@joshblack
joshblack deleted the 3688-remove-wildcard-exports branch October 18, 2024 14:53
@primerprimerBot mentioned this pull request Oct 17, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@broccolinisoup