Skip to content

Add focus styles to Pagination component - #5082

Merged
langermank merged 3 commits into
mainfrom
pagination-bug-fix
Oct 8, 2024
Merged

Add focus styles to Pagination component#5082
langermank merged 3 commits into
mainfrom
pagination-bug-fix

Conversation

@langermank

@langermanklangermank commented Oct 7, 2024

Copy link
Copy Markdown
Contributor

Closes: https://github.com/github/accessibility-audits/issues/8461
Closes: https://github.com/github/accessibility-audits/issues/8056

  • Updates the focus style for the current page
  • Updates the focus style for all items to match
pagination focus styles

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Oct 7, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 81d639f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Oct 7, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.7 KB (-0.01% 🔽)
packages/react/dist/browser.umd.js96.99 KB (-0.04% 🔽)

@langermank
langermank marked this pull request as ready for review October 7, 2024 22:04
@langermank
langermank requested a review from a team as a code ownerOctober 7, 2024 22:04
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5082 October 7, 2024 23:10 Inactive
@langermank
langermank added this pull request to the merge queueOct 8, 2024
Merged via the queue into main with commit 212714cOct 8, 2024
@langermank
langermank deleted the pagination-bug-fix branch October 8, 2024 01:55
@primerprimerBot mentioned this pull request Oct 8, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@jonrohan